chores(pop transform): standard snt init + R func docstrings - #118
Open
EstebanMontandon wants to merge 4 commits into
Open
chores(pop transform): standard snt init + R func docstrings#118EstebanMontandon wants to merge 4 commits into
EstebanMontandon wants to merge 4 commits into
Conversation
sPuntinG
reviewed
Sep 10, 2026
sPuntinG
left a comment
Contributor
There was a problem hiding this comment.
Also great, seems all good!
Question -- just to double check : did you also test run all edited pipelines with their associated edited .ipynb and .r files?
(as needs to be done by hand because, as we know, the pull script mechanism only pull from main not from feature branches, it's easy to forget pieces)
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Second cleaning (no functional changes)
Scripts touched:
snt_dhis2_population_transformation.ipynb
.R files touched:
snt_dhis2_population_transformation.r
additionally unnecessary try: block was removed from the python pipeline snt_dhis2_population_transformation (no functional changes)